Skip to content

[release-v1.16] Replace quay.io/konflux-ci/konflux-vanguard/task-rpms-signature-scan Docker tag with quay.io/konflux-ci/tekton-catalog/task-rpms-signature-scan 0.2 - #1913

Open
red-hat-konflux-kflux-prd-rh02[bot] wants to merge 1 commit into
release-v1.16from
konflux/references-replacement/release-v1.16/quay.io-konflux-ci-konflux-vanguard-task-rpms-signature-scan-replacement
Open

red-hat-konflux-kflux-prd-rh02[bot] wants to merge 1 commit into
release-v1.16from
konflux/references-replacement/release-v1.16/quay.io-konflux-ci-konflux-vanguard-task-rpms-signature-scan-replacement

Conversation

@red-hat-konflux-kflux-prd-rh02

@red-hat-konflux-kflux-prd-rh02 red-hat-konflux-kflux-prd-rh02 Bot commented Jul 25, 2026

Copy link
Copy Markdown

This PR contains the following updates:

Package Type Update Change
quay.io/konflux-ci/konflux-vanguard/task-rpms-signature-scan (source, changelog) → quay.io/konflux-ci/tekton-catalog/task-rpms-signature-scan tekton-bundle replacement 0.20.2

This is a special PR that replaces quay.io/konflux-ci/konflux-vanguard/task-rpms-signature-scan with the community suggested minimal stable replacement version.


Configuration

📅 Schedule: (UTC)

  • Branch creation
    • Only on Saturday (* * * * 6)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test.


Documentation

Find out how to configure dependency updates in MintMaker documentation or see all available configuration options in Renovate documentation.

@openshift-ci
openshift-ci Bot requested review from aliok and lberk July 25, 2026 08:06
@openshift-ci

openshift-ci Bot commented Jul 25, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: red-hat-konflux-kflux-prd-rh02[bot]
Once this PR has been reviewed and has the lgtm label, please assign twogiants for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci

openshift-ci Bot commented Jul 25, 2026

Copy link
Copy Markdown

Hi @red-hat-konflux-kflux-prd-rh02[bot]. Thanks for your PR.

I'm waiting for a openshift-knative member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@red-hat-konflux-kflux-prd-rh02
red-hat-konflux-kflux-prd-rh02 Bot force-pushed the konflux/references-replacement/release-v1.16/quay.io-konflux-ci-konflux-vanguard-task-rpms-signature-scan-replacement branch from 743cdcd to 8475c44 Compare August 8, 2026 08:02
@red-hat-konflux-kflux-prd-rh02 red-hat-konflux-kflux-prd-rh02 Bot changed the title [release-v1.16] Replace quay.io/konflux-ci/konflux-vanguard/task-rpms-signature-scan references [release-v1.16] Replace quay.io/konflux-ci/konflux-vanguard/task-rpms-signature-scan Docker tag with quay.io/konflux-ci/tekton-catalog/task-rpms-signature-scan 0.2 Aug 8, 2026
@red-hat-konflux-kflux-prd-rh02
red-hat-konflux-kflux-prd-rh02 Bot force-pushed the konflux/references-replacement/release-v1.16/quay.io-konflux-ci-konflux-vanguard-task-rpms-signature-scan-replacement branch from 8475c44 to c4b4417 Compare August 15, 2026 08:03
@red-hat-konflux-kflux-prd-rh02
red-hat-konflux-kflux-prd-rh02 Bot force-pushed the konflux/references-replacement/release-v1.16/quay.io-konflux-ci-konflux-vanguard-task-rpms-signature-scan-replacement branch 2 times, most recently from 9543294 to 9cce705 Compare August 29, 2026 08:03
@red-hat-konflux-kflux-prd-rh02
red-hat-konflux-kflux-prd-rh02 Bot force-pushed the konflux/references-replacement/release-v1.16/quay.io-konflux-ci-konflux-vanguard-task-rpms-signature-scan-replacement branch from 9cce705 to fd02073 Compare September 5, 2026 08:04
@red-hat-konflux-kflux-prd-rh02
red-hat-konflux-kflux-prd-rh02 Bot force-pushed the konflux/references-replacement/release-v1.16/quay.io-konflux-ci-konflux-vanguard-task-rpms-signature-scan-replacement branch from fd02073 to f086165 Compare September 12, 2026 08:06
…-signature-scan Docker tag with quay.io/konflux-ci/tekton-catalog/task-rpms-signature-scan 0.2

Signed-off-by: red-hat-konflux-kflux-prd-rh02 <190377777+red-hat-konflux-kflux-prd-rh02[bot]@users.noreply.github.com>
@red-hat-konflux-kflux-prd-rh02
red-hat-konflux-kflux-prd-rh02 Bot force-pushed the konflux/references-replacement/release-v1.16/quay.io-konflux-ci-konflux-vanguard-task-rpms-signature-scan-replacement branch from f086165 to b999be5 Compare September 19, 2026 00:03
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants