Skip to content

[release-v1.17] Update Konflux configurations - #1910

Merged
openshift-merge-bot[bot] merged 2 commits into
openshift-knative:release-v1.17from
serverless-qe:sync-konflux-release-v1.17
Sep 6, 2026
Merged

openshift-merge-bot[bot] merged 2 commits into
openshift-knative:release-v1.17from
serverless-qe:sync-konflux-release-v1.17

Conversation

@serverless-qe

Copy link
Copy Markdown
Collaborator

Update Konflux configurations

@openshift-ci
openshift-ci Bot requested review from Cali0707 and aliok July 17, 2026 11:23
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.17 branch 2 times, most recently from 17ab061 to 02cd3e4 Compare July 20, 2026 09:28
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.17 branch 2 times, most recently from 3dfd8c7 to c8e762c Compare August 3, 2026 10:15
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.17 branch 2 times, most recently from aa77e17 to 67321c3 Compare August 17, 2026 07:25
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.17 branch 5 times, most recently from fa7912c to b084ebf Compare August 26, 2026 14:03
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.17 branch 12 times, most recently from bb9d684 to 43a973b Compare September 2, 2026 16:58
@Kaustubh-pande

Copy link
Copy Markdown

/lgtm
/approve

@openshift-merge-bot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD 785c5f0 and 2 for PR HEAD 43a973b in total

@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.17 branch from 43a973b to a1ed560 Compare September 3, 2026 10:20
@openshift-ci openshift-ci Bot removed the lgtm label Sep 3, 2026
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.17 branch from a1ed560 to 063d103 Compare September 3, 2026 15:02
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.17 branch from 063d103 to 8e77f28 Compare September 3, 2026 18:22
@Kaustubh-pande

Copy link
Copy Markdown

/lgtm
/approve

@openshift-ci openshift-ci Bot added the lgtm label Sep 6, 2026
@openshift-ci

openshift-ci Bot commented Sep 6, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: Kaustubh-pande, serverless-qe

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-bot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD 785c5f0 and 2 for PR HEAD 8e77f28 in total

@openshift-merge-bot
openshift-merge-bot Bot merged commit 9cae548 into openshift-knative:release-v1.17 Sep 6, 2026
35 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants