Skip to content

✨ Use eventdriven wait for cache and eventhandler sync - #3615

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes-sigs:mainfrom
alvaroaleman:eventdriven
Oct 5, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes-sigs:mainfrom
alvaroaleman:eventdriven

Conversation

@alvaroaleman

Copy link
Copy Markdown
Member

This change makes us use the eventdriven HasSyncedChecker instead of polling HasSynced when waiting for the cache and eventhandler sync.

This change makes us use the eventdriven HasSyncedChecker instead of
polling `HasSynced` when waiting for the cache and eventhandler sync.
@kubernetes-prow kubernetes-prow Bot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label Oct 3, 2026
@kubernetes-prow
kubernetes-prow Bot requested review from inteon and sbueringer October 3, 2026 02:08
@kubernetes-prow kubernetes-prow Bot added approved Indicates a PR has been approved by an approver from all required OWNERS files. size/S Denotes a PR that changes 10-29 lines, ignoring generated files. labels Oct 3, 2026
@sbueringer

Copy link
Copy Markdown
Member

Thx!

/lgtm
/approve

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Oct 5, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

LGTM label has been added.

DetailsGit tree hash: 2f560b9c3bb671dcd630607e8fc61480262ea67f

@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: alvaroaleman, sbueringer

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [alvaroaleman,sbueringer]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow
kubernetes-prow Bot merged commit 117d60f into kubernetes-sigs:main Oct 5, 2026
11 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants